Adding in google_memorystore_auth_token, google_memorystore_token_auth_user, google_memorystore_auth_token and google_redis_token_auth_user with corresponding datasources - #18395
Conversation
|
Googlers: For automatic test runs see go/terraform-auto-test-runs. @rileykarson, a repository maintainer, has been assigned to review your changes. If you have not received review feedback within 2 business days, please leave a comment on this PR asking them to take a look. You can help make sure that review is quick by doing a self-review and by running impacted tests locally. |
|
Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 033bff8: Diff reportYour PR generated the following diffs in downstream repositories:
Missing test reportYour PR includes resource fields which are not covered by any test. Resource: resource "google_redis_token_auth_user" "primary" {
region = # value needed
}
Missing service labelsThe following new resources do not have corresponding service labels:
If you believe this detection to be incorrect please raise the concern with your reviewer. Googlers: This error is safe to ignore once you've completed go/fix-missing-service-labels. Multiple resources addedThis PR adds multiple new resources: Missing doc report (experimental)The following data sources are missing documents:
Test reportAnalytics
Affected Service Packages
Step 1: Replaying Mode Action takenFound 12 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit. Click here to see the affected tests
View the replaying VCR build log Step 2: Recording Mode
Caution Issues requiring attention before PR completion 🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details. Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer. View the recording VCR build log or the debug logs folder for detailed results. @NA2047, @rileykarson VCR tests complete for 033bff8! |
|
Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 5d55038: Diff reportYour PR generated the following diffs in downstream repositories:
Missing test reportYour PR includes resource fields which are not covered by any test. Resource: resource "google_redis_cluster_token_auth_user" "primary" {
region = # value needed
}
Multiple resources addedThis PR adds multiple new resources: Test reportAnalytics
Affected Service Packages
Step 1: Replaying Mode Action takenFound 11 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit. Click here to see the affected tests
View the replaying VCR build log Step 2: Recording Mode
Caution Issues requiring attention before PR completion 🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details. Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer. View the recording VCR build log or the debug logs folder for detailed results. @NA2047, @rileykarson VCR tests complete for 5d55038! |
|
Hi there, I'm the Modular magician. I've detected the following information about your changes for commit f191029: Diff reportYour PR generated the following diffs in downstream repositories:
Missing test reportYour PR includes resource fields which are not covered by any test. Resource: resource "google_redis_cluster_token_auth_user" "primary" {
region = # value needed
}
Multiple resources addedThis PR adds multiple new resources: Test reportAnalytics
Affected Service Packages
Step 1: Replaying Mode Action takenFound 11 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit. Click here to see the affected tests
View the replaying VCR build log Step 2: Recording Mode
Caution Issues requiring attention before PR completion 🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details. Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer. View the recording VCR build log or the debug logs folder for detailed results. @NA2047, @rileykarson VCR tests complete for f191029! |
|
Hi there, I'm the Modular magician. I've detected the following information about your changes for commit e004c91: Diff reportYour PR generated the following diffs in downstream repositories:
Missing test reportYour PR includes resource fields which are not covered by any test. Resource: resource "google_redis_cluster_token_auth_user" "primary" {
region = # value needed
}
Multiple resources addedThis PR adds multiple new resources: Test reportAnalytics
Affected Service Packages
Step 1: Replaying Mode Action takenFound 8 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit. Click here to see the affected tests
View the replaying VCR build log Step 2: Recording Mode
Caution Issues requiring attention before PR completion 🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details. Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer. View the recording VCR build log or the debug logs folder for detailed results. @NA2047, @rileykarson VCR tests complete for e004c91! |
|
Hi there, I'm the Modular magician. I've detected the following information about your changes for commit ea12102: Diff reportYour PR generated the following diffs in downstream repositories:
Missing test reportYour PR includes resource fields which are not covered by any test. Resource: resource "google_redis_cluster_token_auth_user" "primary" {
region = # value needed
}
Multiple resources addedThis PR adds multiple new resources: Test reportAnalytics
Affected Service Packages
Step 1: Replaying Mode Action takenFound 8 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit. Click here to see the affected tests
View the replaying VCR build log Step 2: Recording Mode
Caution Issues requiring attention before PR completion 🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details. Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer. View the recording VCR build log or the debug logs folder for detailed results. @NA2047, @rileykarson VCR tests complete for ea12102! |
|
Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 5e3143a: Diff reportYour PR generated the following diffs in downstream repositories:
Missing test reportYour PR includes resource fields which are not covered by any test. Resource: resource "google_redis_cluster_token_auth_user" "primary" {
region = # value needed
}
Multiple resources addedThis PR adds multiple new resources: Test reportAnalytics
Affected Service Packages
Step 1: Replaying Mode Action takenFound 7 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit. Click here to see the affected tests
View the replaying VCR build log Step 2: Recording Mode
Caution Issues requiring attention before PR completion 🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details. Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer. View the recording VCR build log or the debug logs folder for detailed results. @NA2047, @rileykarson VCR tests complete for 5e3143a! |
|
Hi there, I'm the Modular magician. I've detected the following information about your changes for commit fd53d16: Diff reportYour PR generated the following diffs in downstream repositories:
Missing test reportYour PR includes resource fields which are not covered by any test. Resource: resource "google_redis_cluster_token_auth_user" "primary" {
region = # value needed
}
Multiple resources addedThis PR adds multiple new resources: Test reportAnalytics
Affected Service Packages
Step 1: Replaying Mode Action takenFound 7 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit. Click here to see the affected tests
View the replaying VCR build log Step 2: Recording Mode
Caution Issues requiring attention before PR completion 🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details. Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer. View the recording VCR build log or the debug logs folder for detailed results. @NA2047, @rileykarson VCR tests complete for fd53d16! |
|
Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 3948535: Diff reportYour PR generated the following diffs in downstream repositories:
Missing test reportYour PR includes resource fields which are not covered by any test. Resource: resource "google_redis_cluster_token_auth_user" "primary" {
region = # value needed
}
Multiple resources addedThis PR adds multiple new resources: Test reportAnalytics
Affected Service Packages
Step 1: Replaying Mode Action takenFound 7 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit. Click here to see the affected tests
View the replaying VCR build log Step 2: Recording Mode
Caution Issues requiring attention before PR completion 🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details. Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer. View the recording VCR build log or the debug logs folder for detailed results. @NA2047, @rileykarson VCR tests complete for 3948535! |
|
Hi there, I'm the Modular magician. I've detected the following information about your changes for commit fa810ac: Diff reportYour PR generated the following diffs in downstream repositories:
Missing test reportYour PR includes resource fields which are not covered by any test. Resource: resource "google_redis_cluster_token_auth_user" "primary" {
region = # value needed
}
Multiple resources addedThis PR adds multiple new resources: Test reportAnalytics
Affected Service Packages
Step 1: Replaying Mode Action takenFound 7 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit. Click here to see the affected tests
View the replaying VCR build log Step 2: Recording Mode
Caution Issues requiring attention before PR completion 🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details. Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer. View the recording VCR build log or the debug logs folder for detailed results. @NA2047, @rileykarson VCR tests complete for fa810ac! |
|
Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 6567d9d: Diff reportYour PR generated the following diffs in downstream repositories:
Multiple resources addedThis PR adds multiple new resources: Test reportAnalytics
Affected Service Packages
Step 1: Replaying Mode Action takenFound 5 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit. Click here to see the affected tests
View the replaying VCR build log Step 2: Recording Mode
🟢 All tests passed! View the recording VCR build log or the debug logs folder for detailed results. @NA2047, @rileykarson VCR tests complete for 6567d9d! |
Release Note Template for Downstream PRs (will be copied)
See Write release notes for guidance.