Skip to content

feat: respect the destination while packaging frameworks - #733

Open
hurali97 wants to merge 1 commit into
mainfrom
hur/fix-respect-destination-flag
Open

hurali97 wants to merge 1 commit into
mainfrom
hur/fix-respect-destination-flag

Conversation

@hurali97

@hurali97 hurali97 commented Oct 8, 2026

Copy link
Copy Markdown
Contributor

Description

This PR fixes the iOS brownfield packaging to respect --destination when creating the XCFrameworks.

Required for callstack/react-native-brownfield#485

Test Plan

  • Verified on react-native-brownfield repo

@vercel

vercel Bot commented Oct 8, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
rock Ignored Ignored Oct 8, 2026 10:52am UTC

Request Review

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant